Skip to content

[release-v1.23] Update Konflux configurations - #1518

Open
serverless-qe wants to merge 2 commits into
openshift-knative:release-v1.23from
serverless-qe:sync-konflux-release-v1.23
Open

[release-v1.23] Update Konflux configurations#1518
serverless-qe wants to merge 2 commits into
openshift-knative:release-v1.23from
serverless-qe:sync-konflux-release-v1.23

Conversation

@serverless-qe

Copy link
Copy Markdown

Update Konflux configurations

@openshift-ci

openshift-ci Bot commented Aug 24, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign gauron99 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@codecov-commenter

codecov-commenter commented Aug 24, 2026

Copy link
Copy Markdown

⚠️ Please install the 'codecov app svg image' to ensure uploads and comments are reliably processed by Codecov.

Codecov Report

✅ All modified and coverable lines are covered by tests.
⚠️ Please upload report for BASE (release-v1.23@ab17af4). Learn more about missing BASE report.
❗ Your organization needs to install the Codecov GitHub app to enable full functionality.

Additional details and impacted files
@@               Coverage Diff                @@
##             release-v1.23    #1518   +/-   ##
================================================
  Coverage                 ?   44.38%           
================================================
  Files                    ?      203           
  Lines                    ?    18712           
  Branches                 ?        0           
================================================
  Hits                     ?     8306           
  Misses                   ?     9502           
  Partials                 ?      904           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 3 times, most recently from abec9d6 to 51754df Compare August 31, 2026 13:18
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 13 times, most recently from 13031b1 to 47832c1 Compare September 7, 2026 12:00
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch 2 times, most recently from 2f7e4db to ad83052 Compare September 8, 2026 08:54
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.23 branch from ad83052 to c9805a9 Compare September 8, 2026 18:19
@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown

@serverless-qe: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/420-unit c9805a9 link true /test 420-unit
ci/prow/420-images c9805a9 link true /test 420-images
ci/prow/420-e2e-oncluster-test c9805a9 link true /test 420-e2e-oncluster-test

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants