Skip to content

[release-v1.22] Update OWNERS file - #1512

Open
serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.22from
serverless-qe:sync-owners-release-v1.22
Open

serverless-qe wants to merge 1 commit into
openshift-knative:release-v1.22from
serverless-qe:sync-owners-release-v1.22

Conversation

@serverless-qe

Copy link
Copy Markdown

Update OWNERS file

@openshift-ci

openshift-ci Bot commented May 19, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@openshift-ci
openshift-ci Bot requested review from jrangelramos and lkingland May 19, 2026 13:32
@openshift-ci

openshift-ci Bot commented May 19, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign dsimansk for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@codecov-commenter

codecov-commenter commented May 19, 2026

Copy link
Copy Markdown

⚠️ Please install the 'codecov app svg image' to ensure uploads and comments are reliably processed by Codecov.

Codecov Report

✅ All modified and coverable lines are covered by tests.
⚠️ Please upload report for BASE (release-v1.22@fdd1441). Learn more about missing BASE report.
❗ Your organization needs to install the Codecov GitHub app to enable full functionality.

Additional details and impacted files
@@               Coverage Diff                @@
##             release-v1.22    #1512   +/-   ##
================================================
  Coverage                 ?   44.12%           
================================================
  Files                    ?      180           
  Lines                    ?    16060           
  Branches                 ?        0           
================================================
  Hits                     ?     7086           
  Misses                   ?     8137           
  Partials                 ?      837           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from a02ff6c to 95b6751 Compare May 19, 2026 16:37
@openshift-ci

openshift-ci Bot commented May 19, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 95b6751 to 8e5f546 Compare May 20, 2026 12:17
@openshift-ci

openshift-ci Bot commented May 20, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 8e5f546 to 2a48a6b Compare May 20, 2026 15:27
@openshift-ci

openshift-ci Bot commented May 20, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 2a48a6b to 2ac621d Compare May 20, 2026 20:49
@openshift-ci

openshift-ci Bot commented May 20, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 2ac621d to f1c218d Compare May 21, 2026 06:39
@openshift-ci

openshift-ci Bot commented May 21, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from f1c218d to 6f849c4 Compare May 25, 2026 10:50
@openshift-ci

openshift-ci Bot commented May 25, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 6f849c4 to 15072f6 Compare May 29, 2026 14:22
@openshift-ci

openshift-ci Bot commented May 29, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 15072f6 to c6394a0 Compare May 29, 2026 16:48
@openshift-ci

openshift-ci Bot commented May 29, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from c6394a0 to 9f0e136 Compare June 1, 2026 12:16
@openshift-ci

openshift-ci Bot commented Jun 1, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 9f0e136 to d130605 Compare June 2, 2026 04:01
@openshift-ci

openshift-ci Bot commented Jun 2, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from d130605 to b0b7ad4 Compare June 2, 2026 07:20
@openshift-ci

openshift-ci Bot commented Jun 2, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from b0b7ad4 to bd20273 Compare June 2, 2026 11:57
@openshift-ci

openshift-ci Bot commented Jun 2, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from bd20273 to 1cbb721 Compare June 2, 2026 14:43
@openshift-ci

openshift-ci Bot commented Sep 1, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 137570f to bc3951f Compare September 1, 2026 18:57
@openshift-ci

openshift-ci Bot commented Sep 1, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from bc3951f to 865a608 Compare September 1, 2026 22:45
@openshift-ci

openshift-ci Bot commented Sep 1, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 865a608 to 92cf876 Compare September 2, 2026 04:43
@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 92cf876 to 9e7e9e1 Compare September 2, 2026 06:21
@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 9e7e9e1 to 48c3d4e Compare September 2, 2026 13:03
@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 48c3d4e to 8ef5781 Compare September 2, 2026 17:00
@openshift-ci

openshift-ci Bot commented Sep 2, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 8ef5781 to d0ecda7 Compare September 3, 2026 10:22
@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from d0ecda7 to 06dd422 Compare September 3, 2026 15:03
@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 06dd422 to e2831ba Compare September 3, 2026 18:23
@openshift-ci

openshift-ci Bot commented Sep 3, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from e2831ba to 70ebb5e Compare September 7, 2026 12:00
@openshift-ci

openshift-ci Bot commented Sep 7, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 70ebb5e to 2858301 Compare September 7, 2026 18:14
@openshift-ci

openshift-ci Bot commented Sep 7, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 2858301 to 02f8908 Compare September 8, 2026 08:54
@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 02f8908 to 007cfbe Compare September 8, 2026 18:19
@openshift-ci

openshift-ci Bot commented Sep 8, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

@serverless-qe
serverless-qe force-pushed the sync-owners-release-v1.22 branch from 007cfbe to 0475c65 Compare September 14, 2026 12:21
@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown

There are empty aliases in OWNER_ALIASES, cleanup is advised.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants