Skip to content

[release-v1.21] Update Konflux configurations - #1510

Open
serverless-qe wants to merge 2 commits into
openshift-knative:release-v1.21from
serverless-qe:sync-konflux-release-v1.21
Open

serverless-qe wants to merge 2 commits into
openshift-knative:release-v1.21from
serverless-qe:sync-konflux-release-v1.21

Conversation

@serverless-qe

Copy link
Copy Markdown

Update Konflux configurations

@openshift-ci
openshift-ci Bot requested review from creydr and jrangelramos May 19, 2026 13:32
@openshift-ci

openshift-ci Bot commented May 19, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign gauron99 for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@codecov-commenter

codecov-commenter commented May 19, 2026

Copy link
Copy Markdown

⚠️ Please install the 'codecov app svg image' to ensure uploads and comments are reliably processed by Codecov.

Codecov Report

✅ All modified and coverable lines are covered by tests.
⚠️ Please upload report for BASE (release-v1.21@49ecf2e). Learn more about missing BASE report.
❗ Your organization needs to install the Codecov GitHub app to enable full functionality.

Additional details and impacted files
@@               Coverage Diff                @@
##             release-v1.21    #1510   +/-   ##
================================================
  Coverage                 ?   44.37%           
================================================
  Files                    ?      172           
  Lines                    ?    15388           
  Branches                 ?        0           
================================================
  Hits                     ?     6828           
  Misses                   ?     7770           
  Partials                 ?      790           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 6 times, most recently from 9d95de1 to 4cec55c Compare May 25, 2026 10:50
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 10 times, most recently from 965a70a to 37b4c17 Compare June 4, 2026 12:45
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 2 times, most recently from 25c78ae to 6e68bbd Compare June 10, 2026 20:02
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 2 times, most recently from aa49925 to 3e907bf Compare June 22, 2026 12:24
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 2 times, most recently from c2c6968 to e579bc0 Compare June 29, 2026 11:29
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 3 times, most recently from 3501785 to c52c52b Compare July 9, 2026 07:50
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch from c52c52b to 048be1d Compare July 13, 2026 09:40
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 2 times, most recently from afe8809 to aa10475 Compare August 17, 2026 07:26
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 4 times, most recently from 005a052 to 93ccc23 Compare August 26, 2026 14:05
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 14 times, most recently from e4002df to 72492c2 Compare September 7, 2026 12:00
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 4 times, most recently from 6cf02de to dc1b244 Compare September 14, 2026 12:21
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch 2 times, most recently from 5874a7a to 4ee744d Compare September 15, 2026 11:46
@serverless-qe
serverless-qe force-pushed the sync-konflux-release-v1.21 branch from 4ee744d to 1adddd1 Compare September 15, 2026 13:11
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown

@serverless-qe: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/421-e2e-oncluster-test 1adddd1 link true /test 421-e2e-oncluster-test

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants