Skip to content

Add execution buffer max size - #491

Merged
openscript merged 5 commits into
mainfrom
add-execution-buffer
Sep 22, 2025
Merged

openscript merged 5 commits into
mainfrom
add-execution-buffer

Conversation

@openscript

Copy link
Copy Markdown
Member

Fixes #446 regression

@changeset-bot

changeset-bot Bot commented Sep 22, 2025 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f5e52a2

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@quassel/frontend Patch
@quassel/backend Patch
@quassel/website Patch
@quassel/utils Patch
@quassel/ui Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@openscript openscript changed the title Add execution buffer Add execution buffer max size Sep 22, 2025
@openscript
openscript merged commit 3dbdf91 into main Sep 22, 2025
1 check passed
@openscript
openscript deleted the add-execution-buffer branch September 22, 2025 12:32
@codecov

codecov Bot commented Sep 22, 2025

Copy link
Copy Markdown

Codecov Report

❌ Patch coverage is 0% with 2 lines in your changes missing coverage. Please review.
✅ Project coverage is 16.90%. Comparing base (3c745f5) to head (f5e52a2).
⚠️ Report is 6 commits behind head on main.

Files with missing lines Patch % Lines
apps/backend/src/config/configuration.ts 0.00% 1 Missing ⚠️
apps/backend/src/system/export/export.service.ts 0.00% 1 Missing ⚠️
Additional details and impacted files
@@            Coverage Diff             @@
##             main     #491      +/-   ##
==========================================
- Coverage   16.90%   16.90%   -0.01%     
==========================================
  Files         132      132              
  Lines        3992     3993       +1     
  Branches      129      136       +7     
==========================================
  Hits          675      675              
- Misses       3257     3258       +1     
  Partials       60       60              

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Error during export

1 participant