Skip to content

update structure arg .Names to names for R-devel - #96

Merged
nealrichardson merged 3 commits into
nealrichardson:mainfrom
gergness:structure-deprecations
Jul 31, 2026
Merged

nealrichardson merged 3 commits into
nealrichardson:mainfrom
gergness:structure-deprecations

Conversation

@gergness

Copy link
Copy Markdown
Contributor

Happy Summer!

This PR fixes the CRAN notes about r-devel's upcoming deprecations to structure(.Names =...). It snuck up on me for rcrunch, so wanted to help upstream. I've skipped a few tests that are failing in rcrunch because expect_silent() is catching warnings from httptest for this.

Thanks!

@capnrefsmmat

Copy link
Copy Markdown

Heh, thanks for getting to this before I could finish (#95). Since I already had an R-devel test environment set up, I pulled your branch to test. Can confirm there are no deprecation warnings anymore.

But I do see:

Failure (test-expect-header.R:120:5): expect_header works with actual network too
Expected 0 successes and 1 failure.
x Observed 1 success.
v Observed 1 failure.

I see that's already set to skip on CI, so maybe that's not an issue.

@gergness

Copy link
Copy Markdown
Contributor Author

Ha, didn't realize CRAN did the nastygrams on a rolling schedule like that (crunch had until July 25th). Thanks for checking!

@nealrichardson

Copy link
Copy Markdown
Owner

Thank you for this! I have fixed CI on main, would you mind rebasing to pull those changes in, and then here bump the version number and add a NEWS entry? Then I can cut a release.

@gergness
gergness force-pushed the structure-deprecations branch from c7f93ea to d5add84 Compare July 30, 2026 15:42
@codecov-commenter

Copy link
Copy Markdown

⚠️ Please install the 'codecov app svg image' to ensure uploads and comments are reliably processed by Codecov.

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 98.03%. Comparing base (d7de84e) to head (0e48e5e).
⚠️ Report is 4 commits behind head on main.
❗ Your organization needs to install the Codecov GitHub app to enable full functionality.

Additional details and impacted files
@@             Coverage Diff             @@
##              main      #96      +/-   ##
===========================================
- Coverage   100.00%   98.03%   -1.97%     
===========================================
  Files           18       18              
  Lines          508      508              
===========================================
- Hits           508      498      -10     
- Misses           0       10      +10     

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@nealrichardson nealrichardson left a comment

Copy link
Copy Markdown
Owner

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thank you!

@nealrichardson
nealrichardson merged commit 41bc2b9 into nealrichardson:main Jul 31, 2026
13 checks passed
@capnrefsmmat

Copy link
Copy Markdown

Now that CRAN is back from summer vacation, will it be possible to submit a new release soon?

I can work around the issue for my package by submitting a release that skips the affected tests, but it'd be good to know either way so I can prep that if I have to. They gave me until Aug 21, although maybe they'd give leeway if they know the issue is fixed and incoming to CRAN.

@nealrichardson

Copy link
Copy Markdown
Owner

Thanks for the reminder, I'll try to get submitted in the next day now that they're open again.

@nealrichardson

Copy link
Copy Markdown
Owner

Release is up on CRAN now. Thanks for your help with this.

@capnrefsmmat

Copy link
Copy Markdown

Thanks! CRAN ran a check run on my package this morning and it seems to have passed, so the fix worked.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants