Skip to content

fix(getting-started): Fix typos in systemd service config - #541

Merged
f2c-ci-robot[bot] merged 1 commit into
halo-dev:mainfrom
HowieHz:fix-536
Dec 25, 2025
Merged

f2c-ci-robot[bot] merged 1 commit into
halo-dev:mainfrom
HowieHz:fix-536

Conversation

@HowieHz

@HowieHz HowieHz commented Dec 24, 2025 •

Copy link
Copy Markdown
Contributor

fix #536

None

@f2c-ci-robot f2c-ci-robot Bot added the do-not-merge/release-note-label-needed Indicates that a PR should not merge because it's missing one of the release note labels. label Dec 24, 2025

@ruibaby ruibaby left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@f2c-ci-robot f2c-ci-robot Bot added the lgtm Indicates that a PR is ready to be merged. label Dec 25, 2025
@f2c-ci-robot

f2c-ci-robot Bot commented Dec 25, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: ruibaby

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@f2c-ci-robot f2c-ci-robot Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. release-note-none Denotes a PR that doesn't merit a release note. and removed do-not-merge/release-note-label-needed Indicates that a PR should not merge because it's missing one of the release note labels. labels Dec 25, 2025
@f2c-ci-robot
f2c-ci-robot Bot merged commit ac203d7 into halo-dev:main Dec 25, 2025
2 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. release-note-none Denotes a PR that doesn't merit a release note.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

[Docs] Fix typos in Systemd service configuration template

2 participants