Skip to content

docs: update documentations for 2.22 - #537

Merged
ruibaby merged 1 commit into
halo-dev:mainfrom
ruibaby:docs/2.22
Dec 24, 2025
Merged

ruibaby merged 1 commit into
halo-dev:mainfrom
ruibaby:docs/2.22

Conversation

@ruibaby

@ruibaby ruibaby commented Dec 24, 2025

Copy link
Copy Markdown
Member
None

Signed-off-by: Ryan Wang <i@ryanc.cc>
@f2c-ci-robot f2c-ci-robot Bot added the release-note-none Denotes a PR that doesn't merit a release note. label Dec 24, 2025
@f2c-ci-robot

f2c-ci-robot Bot commented Dec 24, 2025

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign guqing for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@ruibaby
ruibaby merged commit 5953f17 into halo-dev:main Dec 24, 2025
1 of 2 checks passed
@ruibaby
ruibaby deleted the docs/2.22 branch December 24, 2025 03:31
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

release-note-none Denotes a PR that doesn't merit a release note.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant