Skip to content

chore: use synced teams in CODEOWNERS - #484

Merged
crowecawcaw merged 2 commits into
aws-deadline:mainlinefrom
crowecawcaw:chore/codeowners-sync-teams
Oct 2, 2026
Merged

crowecawcaw merged 2 commits into
aws-deadline:mainlinefrom
crowecawcaw:chore/codeowners-sync-teams

Conversation

@crowecawcaw

Copy link
Copy Markdown
Contributor

Replace maya-developers with sync-team-aws-deadline-cloud-github-developers in CODEOWNERS.

Signed-off-by: Stephen Crowe <6042774+crowecawcaw@users.noreply.github.com>
@crowecawcaw
crowecawcaw requested a review from a team as a code owner October 1, 2026 17:10
@github-actions github-actions Bot added the waiting-on-maintainers Waiting on the maintainers to review. label Oct 1, 2026
@crowecawcaw
crowecawcaw enabled auto-merge (squash) October 1, 2026 17:37
@crowecawcaw crowecawcaw changed the title chore: use synced teams in CODEOWNERS chore: use synced teams in CODEOWNERS Oct 1, 2026
@crowecawcaw crowecawcaw changed the title chore: use synced teams in CODEOWNERS chore: use synced teams in CODEOWNERS Oct 1, 2026
@crowecawcaw
crowecawcaw merged commit 0e47968 into aws-deadline:mainline Oct 2, 2026
17 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

waiting-on-maintainers Waiting on the maintainers to review.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants