Skip to content

ci: Remove the Run Samples workflow - #298

Merged
kdestin merged 1 commit into
Azure-Samples:mainfrom
kdestin:remove-run-samples
Oct 6, 2026
Merged

kdestin merged 1 commit into
Azure-Samples:mainfrom
kdestin:remove-run-samples

Conversation

@kdestin

@kdestin kdestin commented Oct 2, 2026 •

Copy link
Copy Markdown
Member

Description

This pull request disables the run-samples.yml workflow. This addresses a security scan that flagged our use of pull_request_target. run-samples.yml did include the github-recommended mitigation against pwning pull_request_target workflows. Disabling instead of reworking since this repository has largely been superseded.

Checklist

  • I have read the contribution guidelines
  • I have coordinated with the docs team (mldocs@microsoft.com) if this PR deletes files or changes any file names or file extensions.
  • This notebook or file is added to the CODEOWNERS file, pointing to the author or the author’s team.

@kdestin
kdestin requested a review from a team as a code owner October 2, 2026 19:56
@kdestin
kdestin enabled auto-merge (squash) October 2, 2026 21:19
@kdestin
kdestin disabled auto-merge October 6, 2026 18:28
@kdestin
kdestin merged commit 3e77be8 into Azure-Samples:main Oct 6, 2026
3 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants